Skip to content

Adding 26.3 support - #81

Merged
Ale32bit merged 1 commit into
mainfrom
26.3
Sep 24, 2026
Merged

Ale32bit merged 1 commit into
mainfrom
26.3

Conversation

@nelsongraca

Copy link
Copy Markdown
Collaborator

Not much has changed.

Affected modules/commands:

  • RocketCommand
  • NudgeCommand
  • DoAsCommand
  • SudoCommand
  • SignModule
  • AdvancementFormatter

@nelsongraca
nelsongraca marked this pull request as ready for review September 23, 2026 12:08
@cloudflare-workers-and-pages

cloudflare-workers-and-pages Bot commented Sep 23, 2026 •

Copy link
Copy Markdown

Deploying solstice with  Cloudflare Pages  Cloudflare Pages

Latest commit: 747725f
Status: ✅  Deploy successful!
Preview URL: https://0ec029a3.solstice-29r.pages.dev
Branch Preview URL: https://26-3.solstice-29r.pages.dev

View logs

@nelsongraca

Copy link
Copy Markdown
Collaborator Author

Tested affected modules, FormatSignMixin had to be adjusted.

@Ale32bit
Ale32bit merged commit 3f39d93 into main Sep 24, 2026
7 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants